Skip to content

Migrate fixtures to nitro v3, bump srvx - #2188

Merged
birkskyum merged 5 commits into
mainfrom
use-nitro-v3-in-fixtures
Jul 16, 2026
Merged

birkskyum merged 5 commits into
mainfrom
use-nitro-v3-in-fixtures

Conversation

@birkskyum

@birkskyum birkskyum commented Jul 16, 2026

Copy link
Copy Markdown
Member

It's much more important to test against nitro v3, so this updates the fixtures

removes redundant nitro-3 fixture

@changeset-bot

changeset-bot Bot commented Jul 16, 2026

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 867f0be

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@solidjs/start Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@netlify

netlify Bot commented Jul 16, 2026

Copy link
Copy Markdown

Deploy Preview for solid-start-landing-page ready!

Name Link
🔨 Latest commit 867f0be
🔍 Latest deploy log https://app.netlify.com/projects/solid-start-landing-page/deploys/6a5848298fc01f00089f1edd
😎 Deploy Preview https://deploy-preview-2188--solid-start-landing-page.netlify.app
📱 Preview on mobile
Toggle QR Code...

QR Code

Use your smartphone camera to open QR code link.
🤖 Make changes Run an agent on this branch

To edit notification comments on pull requests, go to your Netlify project configuration.

@pkg-pr-new

pkg-pr-new Bot commented Jul 16, 2026

Copy link
Copy Markdown

Open in StackBlitz

npm i https://pkg.pr.new/solidjs/solid-start/@solidjs/start@2188
npm i https://pkg.pr.new/solidjs/solid-start/@solidjs/vite-plugin-nitro-2@2188

commit: 867f0be

@birkskyum
birkskyum requested a review from brenelz July 16, 2026 02:34
@birkskyum
birkskyum requested review from katywings and lxsmnsyc July 16, 2026 02:39
@brenelz

brenelz commented Jul 16, 2026

Copy link
Copy Markdown
Contributor

Landing page deploy preview broken

https://deploy-preview-2188--solid-start-landing-page.netlify.app/

@birkskyum birkskyum changed the title Migrate fixtures to nitro v3 Migrate fixtures to nitro v3, bump srvx Jul 16, 2026
@birkskyum
birkskyum merged commit ee1193a into main Jul 16, 2026
10 checks passed
@birkskyum
birkskyum deleted the use-nitro-v3-in-fixtures branch July 16, 2026 02:57
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants