chore(deps): update bgp-cloud-connector-operator to d64d026 - #93
Conversation
|
Pipeline controller notification For optional jobs, comment This repository is configured in: LGTM mode |
|
Hi @red-hat-konflux-kflux-prd-rh02[bot]. Thanks for your PR. I'm waiting for a openshift member to verify that this patch is reasonable to test. If it is, they should reply with Regular contributors should join the org to skip this step. Once the patch is verified, the new status will be reflected by the I understand the commands that are listed here. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
4361b5a to
70f67d6
Compare
a321599 to
0ac5a1a
Compare
0ac5a1a to
6a8caf7
Compare
6a8caf7 to
6be7695
Compare
|
/ok-to-test |
6be7695 to
0cf2b83
Compare
0cf2b83 to
1e26c05
Compare
1e26c05 to
fa6a24a
Compare
fa6a24a to
a9c8324
Compare
a9c8324 to
b158233
Compare
b158233 to
0223715
Compare
0223715 to
a7e1eab
Compare
a7e1eab to
529e78a
Compare
|
/lgtm |
|
Pipeline controller notification No second-stage tests were triggered for this PR. This can happen when:
Use |
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: alebedev87, red-hat-konflux-kflux-prd-rh02[bot] The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
529e78a to
88fc91d
Compare
88fc91d to
32a74d0
Compare
|
/lgtm |
|
Pipeline controller notification No second-stage tests were triggered for this PR. This can happen when:
Use |
32a74d0 to
762f302
Compare
Image created from 'https://github.com/openshift/bgp-cloud-connector?rev=54b3b3591a561593989310c24923a4f46f40e8db' Signed-off-by: red-hat-konflux-kflux-prd-rh02 <190377777+red-hat-konflux-kflux-prd-rh02[bot]@users.noreply.github.com>
762f302 to
1be9c5b
Compare
|
@red-hat-konflux-kflux-prd-rh02[bot]: all tests passed! Full PR test history. Your PR dashboard. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here. |
|
/lgtm |
Image created from 'https://github.com/openshift/bgp-cloud-connector?rev=54b3b3591a561593989310c24923a4f46f40e8db'
This PR contains the following updates:
d258edc->d64d026Configuration
📅 Schedule: Branch creation - At any time (no schedule defined), Automerge - At any time (no schedule defined).
🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.
♻ Rebasing: Whenever PR is behind base branch, or you tick the rebase/retry checkbox.
👻 Immortal: This PR will be recreated if closed unmerged. Get config help if that's undesired.
To execute skipped test pipelines write comment
/ok-to-testThis is Konflux nudging PR, not Mintmaker.