Repository navigation
moved the breaking changes section to under v37.0.0 - #2728
Conversation
|
/assign @yurnov |
|
@yliaog: GitHub didn't allow me to assign the following users: yurnov. Note that only kubernetes-client members with read permissions, repo collaborators and people who have commented on this issue/PR can be assigned. Additionally, issues/PRs can only have 10 assignees at the same time. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
|
/assign @roycaihw |
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: roycaihw, yliaog The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
1baff76
into
kubernetes-client:release-37.0
What type of PR is this?
/kind documentation
What this PR does / why we need it:
moved the breaking changes section to under v37.0.0
Which issue(s) this PR fixes:
Fixes #
#2726
Special notes for your reviewer:
Does this PR introduce a user-facing change?
Additional documentation e.g., KEPs (Kubernetes Enhancement Proposals), usage docs, etc.: